Skip to content

Generate release testdata files - #10879

Open
eks-distro-pr-bot wants to merge 1 commit into
aws:mainfrom
eks-distro-pr-bot:update-release-test-file
Open

eks-distro-pr-bot wants to merge 1 commit into
aws:mainfrom
eks-distro-pr-bot:update-release-test-file

Conversation

@eks-distro-pr-bot

Copy link
Copy Markdown
Contributor

Generate release testdata files.

By submitting this pull request, I confirm that you can use, modify, copy, and redistribute this contribution, under the terms of your choice.

@eks-distro-bot eks-distro-bot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. needs-ok-to-test labels Jul 21, 2026
@eks-distro-bot

Copy link
Copy Markdown
Collaborator

Hi @eks-distro-pr-bot. Thanks for your PR.

I'm waiting for a aws member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Tip

We noticed you've done this a few times! Consider joining the org to skip this step and gain /lgtm and other bot rights. We recommend asking approvers on your previous PRs to sponsor you.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-release-test-file branch 27 times, most recently from 99ddbfd to f81bb95 Compare July 22, 2026 23:14
@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-release-test-file branch 24 times, most recently from 56112d4 to 1a59cec Compare July 24, 2026 17:14
@peirulu

peirulu commented Jul 24, 2026

Copy link
Copy Markdown
Member

/lgtm
/approve

@eks-distro-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: peirulu

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@peirulu

peirulu commented Jul 24, 2026

Copy link
Copy Markdown
Member

/ok-to-test

@peirulu

peirulu commented Jul 24, 2026

Copy link
Copy Markdown
Member

/lgtm

@eks-distro-bot

Copy link
Copy Markdown
Collaborator

New changes are detected. LGTM label has been removed.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved ok-to-test size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants